Skip to content

chore(deps): update all non-major dependencies - #129

Open
renovate[bot] wants to merge 1 commit into
mainfrom
renovate/all-minor-patch
Open

renovate[bot] wants to merge 1 commit into
mainfrom
renovate/all-minor-patch

Conversation

@renovate

@renovate renovate Bot commented Jun 10, 2024 •

Copy link
Copy Markdown
Contributor

ℹ️ Note

This PR body was truncated due to platform limits.

This PR contains the following updates:

Package Change Age Confidence Type Update
@types/jest (source) 29.5.12 → 29.5.14 age confidence devDependencies patch
commitizen 4.3.0 → 4.3.2 age confidence devDependencies patch
husky 9.0.11 → 9.1.7 age confidence devDependencies minor
node (source) 24.16.0 → 24.21.0 age confidence volta minor
npm-run-all2 5.0.0 → 5.0.2 age confidence devDependencies patch
rimraf 5.0.5 → 5.0.10 age confidence devDependencies patch
semantic-release 25.0.5 → 25.0.9 age confidence devDependencies patch
ts-jest (source) 29.1.2 → 29.4.14 age confidence devDependencies minor
typescript (source) 5.4.2 → 5.9.3 age confidence devDependencies minor
yarn 1.22.21 → 1.22.22 age confidence volta patch

Release Notes

commitizen/cz-cli (commitizen)

v4.3.2

Compare Source

4.3.2 (2026-06-12)

v4.3.1

Compare Source

4.3.1 (2024-09-27)
Bug Fixes
typicode/husky (husky)

v9.1.7

Compare Source

What's Changed

New Contributors

Full Changelog: typicode/husky@v9.1.6...v9.1.7

v9.1.6

Compare Source

What's Changed

New Contributors

Full Changelog: typicode/husky@v9.1.5...v9.1.6

v9.1.5

Compare Source

What's Changed

New Contributors

Full Changelog: typicode/husky@v9.1.4...v9.1.5

v9.1.4

Compare Source

  • Improve deprecation notice

v9.1.3

Compare Source

  • fix: better handle space in PATH

v9.1.2

Compare Source

Show a message instead of automatically removing deprecated code.

This only concerns projects that still have the following code in their hooks:

- #!/usr/bin/env sh # <- This is deprecated, remove it
- . "$(dirname -- "$0")/_/husky.sh"  # <- This is deprecated, remove it

# Rest of your hook code

Hooks with these lines will fail in v10.0.0

v9.1.1

Compare Source

Super saiyan god dog! It's over 9.0.0!

What's new

You can now run package commands directly, no need for npx or equivalents.
It makes writing hooks more intuitive and is also slightly faster 🐺⚡️

# .husky/pre-commit
- npx jest
+ jest # ~0.2s faster

A new recipe has been added to the docs. Lint staged files without external dependencies (inspired by Prettier docs). Feel free to modify it.

# .husky/pre-commit
prettier $(git diff --cached --name-only --diff-filter=ACMR | sed 's| |\\ |g') --write --ignore-unknown
git update-index --again

For more advanced use cases, see lint-staged.

Fixes

  • bunx husky init command
  • Workaround for some hooks implementation on Windows

Deprecations

  • #!/usr/bin/env sh and . "$(dirname -- "$0")/_/husky.sh" are deprecated. husky command will automatically remove them, no action required.
  • If you're having code in ~/.huskyrc please move it to .config/husky/init.sh

Support for these will be removed in v10.

Friendly reminder

If Git hooks don't fit your workflow, you can disable Husky globally. Just add export HUSKY=0 to .config/husky/init.sh.

I've seen some confusion about this on X, so just a heads-up!

Sponsoring

Husky is downloaded over 45M times per month and used by ~1.5M projects. If your company wants to sponsor, you can do so here: GitHub Sponsors.

Have a nice summer ☀️ I'm open to new opportunities/consulting so feel free to drop me a message 😉

v9.1.0

Compare Source

nodejs/node (node)

v24.21.0: 2026-09-08, Version 24.21.0 'Krypton' (LTS), @​aduh95

Compare Source

Notable Changes
  • [71106e1f17] - crypto: update root certificates to NSS 3.126 (Node.js GitHub Bot) #​65495
  • [afca0a912d] - (SEMVER-MINOR) crypto: support loading private keys through STORE loaders (Filip Skokan) #​63949
  • [6274fccbd9] - deps: update OpenSSL to 3.5.8 (Node.js GitHub Bot) #​65542
  • [53cba013c7] - deps: update Undici to 7.29.1 (Node.js GitHub Bot) #​65789
  • [0529772798] - (SEMVER-MINOR) lib,src: improve histogram implementation (James M Snell) #​65024
  • [41c7062b81] - (SEMVER-MINOR) net: improve performance of net.BlockList (James M Snell) #​64974
  • [5197b5a3c5] - (SEMVER-MINOR) perf_hooks: add statistical hypothesis testing to histogram (James M Snell) #​65416
  • [35c635b032] - (SEMVER-MINOR) util: add non-throwing MIMEType.parse (James M Snell) #​64965
Commits

❗ Important

✂ PR body was truncated to here.


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • Between 12:00 AM and 03:59 AM, only on Monday (* 0-3 * * 1)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 2 times, most recently from 6f3a452 to 6f9a460 Compare June 21, 2024 23:31
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 2 times, most recently from af82c68 to 4225eba Compare July 6, 2024 14:49
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 2 times, most recently from ad451f0 to 94b48fa Compare July 12, 2024 05:19
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 4 times, most recently from 3feaef8 to d3276ae Compare July 24, 2024 23:43
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 5 times, most recently from fe99052 to 3a56404 Compare August 2, 2024 02:20
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 2 times, most recently from 7eaed6d to 23cf904 Compare August 23, 2024 20:44
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 3 times, most recently from f8180ca to 20af759 Compare September 13, 2024 17:38
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from 20af759 to 5adc504 Compare September 27, 2024 05:18
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 2 times, most recently from 4f4eb42 to a7113f3 Compare October 9, 2024 05:21
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from a7113f3 to bf41060 Compare October 23, 2024 23:51
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 3 times, most recently from ab872ce to 7191cae Compare November 23, 2024 05:53
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from 7191cae to 0ec656e Compare January 9, 2025 03:34
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from 0ec656e to 038877b Compare January 24, 2025 08:10
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from 038877b to 3a3fc9a Compare February 13, 2025 03:59
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from 39bf69b to 0924ca1 Compare July 26, 2025 12:04
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from 0924ca1 to 0857d73 Compare August 4, 2025 19:37
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from 0857d73 to e1d12ee Compare August 14, 2025 20:10
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from e1d12ee to a6964ef Compare August 23, 2025 19:24
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 2 times, most recently from 557892e to c86d201 Compare September 4, 2025 07:16
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 3 times, most recently from a63d671 to 6a4d42f Compare September 20, 2025 07:30
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 2 times, most recently from b9468fa to 45271a5 Compare October 1, 2025 06:42
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from 45271a5 to db7e56d Compare October 11, 2025 00:07
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from db7e56d to 88efbe3 Compare October 25, 2025 20:09
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 2 times, most recently from 8e8ce29 to 7e7769b Compare November 19, 2025 07:08
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 2 times, most recently from 8a131b6 to 75bc252 Compare December 2, 2025 00:05
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 2 times, most recently from 1fe63c6 to c357bf7 Compare January 19, 2026 23:33
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from c357bf7 to 34c2d71 Compare January 24, 2026 07:14
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from 34c2d71 to 0bdd501 Compare February 3, 2026 23:43
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 2 times, most recently from 68b4eba to 87469c8 Compare February 19, 2026 03:42
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch 2 times, most recently from 221cf3a to fbfec0d Compare March 14, 2026 11:18
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from fbfec0d to 8165b6a Compare March 31, 2026 17:07
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from 8165b6a to c181dc0 Compare April 15, 2026 14:14
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from c181dc0 to 27af7c5 Compare May 1, 2026 11:06
@renovate
renovate Bot force-pushed the renovate/all-minor-patch branch from 27af7c5 to b3201f6 Compare May 14, 2026 02:45
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants