Skip to content

Update release-team-leads Slack group members for v1.38 release cycle - #9138

Open
sreeram-venkitesh wants to merge 2 commits into
kubernetes:mainfrom
sreeram-venkitesh:update-v1.38-release-team
Open

sreeram-venkitesh wants to merge 2 commits into
kubernetes:mainfrom
sreeram-venkitesh:update-v1.38-release-team

Conversation

@sreeram-venkitesh

Copy link
Copy Markdown
Member

Updates the membership of the release-team-leads Slack group for v1.38 release cycle.

Ref: kubernetes/sig-release#3087

@kubernetes-prow kubernetes-prow Bot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Sep 3, 2026
@kubernetes-prow kubernetes-prow Bot added area/community-management area/slack-management Issues or PRs related to the Slack Management subproject size/S Denotes a PR that changes 10-29 lines, ignoring generated files. sig/contributor-experience Categorizes an issue or PR as relevant to SIG Contributor Experience. sig/release Categorizes an issue or PR as relevant to SIG Release. labels Sep 3, 2026
@sreeram-venkitesh

Copy link
Copy Markdown
Member Author

/cc katcosgrove fsmunoz

@kubernetes-prow
kubernetes-prow Bot requested a review from fsmunoz September 3, 2026 17:51
@katcosgrove

Copy link
Copy Markdown
Contributor

/lgtm
/assign @coderanger

@kubernetes-prow kubernetes-prow Bot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Sep 8, 2026
@sreeram-venkitesh

Copy link
Copy Markdown
Member Author

/cc @jberkus would you be able to approve here please?

@palnabarun palnabarun left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/approve

@kubernetes-prow kubernetes-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 16, 2026
@palnabarun

Copy link
Copy Markdown
Member

/hold to check an username

@kubernetes-prow kubernetes-prow Bot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Sep 16, 2026
Comment thread communication/slack-config/users.yaml Outdated
Comment thread communication/slack-config/users.yaml Outdated
Comment thread communication/slack-config/users.yaml Outdated
@kubernetes-prow

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: sreeram-venkitesh
Once this PR has been reviewed and has the lgtm label, please ask for approval from coderanger. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubernetes-prow kubernetes-prow Bot removed the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 16, 2026
@kubernetes-prow kubernetes-prow Bot removed the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Sep 20, 2026
@kubernetes-prow

Copy link
Copy Markdown
Contributor

New changes are detected. LGTM label has been removed.

@sreeram-venkitesh

Copy link
Copy Markdown
Member Author

@palnabarun Please take another look, fixed the user keys.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

area/community-management area/slack-management Issues or PRs related to the Slack Management subproject cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. sig/contributor-experience Categorizes an issue or PR as relevant to SIG Contributor Experience. sig/release Categorizes an issue or PR as relevant to SIG Release. size/S Denotes a PR that changes 10-29 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants