Skip to content

Fix non-contant format string err in go 1.26 - #9406

Merged
knative-prow[bot] merged 1 commit into
knative:mainfrom
dsimansk:pr/fix-format-string
Sep 18, 2026
Merged

knative-prow[bot] merged 1 commit into
knative:mainfrom
dsimansk:pr/fix-format-string

Conversation

@dsimansk

Copy link
Copy Markdown
Contributor

Fixes #

Proposed Changes

  • Fix non-contant format string err in go 1.26

Pre-review Checklist

  • At least 80% unit test coverage
  • E2E tests for any new behavior
  • Docs PR for any user-facing impact
  • Spec PR for any new API feature
  • Conformance test for any change to the spec

Release Note

Fix non-contant format string err in go 1.26

Clean up releated to go vet errors flagged in:
knative/pkg#3392

/cc @gauron99

@knative-prow
knative-prow Bot requested a review from gauron99 September 18, 2026 15:18
@knative-prow knative-prow Bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Sep 18, 2026
@dsimansk

Copy link
Copy Markdown
Contributor Author

/cc @creydr @Cali0707

@knative-prow
knative-prow Bot requested review from Cali0707 and creydr September 18, 2026 15:19
@knative-prow knative-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 18, 2026
@gauron99

Copy link
Copy Markdown

/lgtm looks like a simple and general enough of a fix

@codecov

codecov Bot commented Sep 18, 2026 •

Copy link
Copy Markdown

Codecov Report

❌ Patch coverage is 18.94737% with 77 lines in your changes missing coverage. Please review.
✅ Project coverage is 51.16%. Comparing base (07b1e8a) to head (5b4c1cc).
⚠️ Report is 1 commits behind head on main.

Files with missing lines Patch % Lines
pkg/reconciler/testing/v1/trigger.go 0.00% 10 Missing ⚠️
pkg/reconciler/testing/v1/inmemorychannel.go 0.00% 8 Missing ⚠️
...apis/eventing/v1alpha1/eventtransform_lifecycle.go 0.00% 6 Missing ⚠️
pkg/reconciler/testing/v1/broker.go 0.00% 6 Missing ⚠️
.../apis/sinks/v1alpha1/integration_sink_lifecycle.go 0.00% 5 Missing ⚠️
pkg/reconciler/testing/v1/channel.go 0.00% 5 Missing ⚠️
pkg/apis/eventing/v1/trigger_lifecycle.go 33.33% 4 Missing ⚠️
pkg/reconciler/testing/v1/parallel.go 0.00% 4 Missing ⚠️
pkg/reconciler/testing/v1/sequence.go 0.00% 4 Missing ⚠️
pkg/reconciler/testing/v1/subscription.go 0.00% 4 Missing ⚠️
... and 14 more
Additional details and impacted files
@@            Coverage Diff             @@
##             main    #9406      +/-   ##
==========================================
- Coverage   51.18%   51.16%   -0.03%     
==========================================
  Files         411      411              
  Lines       22176    22173       -3     
==========================================
- Hits        11351    11345       -6     
  Misses       9951     9951              
- Partials      874      877       +3     

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@dsimansk

Copy link
Copy Markdown
Contributor Author

/override "codecov/project"

@knative-prow

knative-prow Bot commented Sep 18, 2026

Copy link
Copy Markdown

@dsimansk: Overrode contexts on behalf of dsimansk: codecov/project

Details

In response to this:

/override "codecov/project"

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@Cali0707 Cali0707 left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@knative-prow knative-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 18, 2026
@knative-prow

knative-prow Bot commented Sep 18, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Cali0707, dsimansk

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow
knative-prow Bot merged commit 99df328 into knative:main Sep 18, 2026
37 of 38 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants