Skip to content

feat: support passing options to reanimated / worklets plugin - #1269

Merged
jbroma merged 6 commits into
mainfrom
feat/allow-passing-options-to-reanimated
Aug 22, 2025
Merged

jbroma merged 6 commits into
mainfrom
feat/allow-passing-options-to-reanimated

Conversation

@jbroma

@jbroma jbroma commented Aug 22, 2025

Copy link
Copy Markdown
Contributor

Summary

This PR adds a way to pass custom options to Reanimated/Worklets plugin.
It also adds a check to ensure react-native-worklets is installed when using Reanimated >= 4

Example:

new ReanimatedPlugin({
  babelPluginOptions: { 
	relativeSourceLocation: true 
  },
}),

Test plan

  • - testers work and options are passed

@changeset-bot

changeset-bot Bot commented Aug 22, 2025 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 5a0638c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 6 packages
Name Type
@callstack/repack-plugin-reanimated Major
@callstack/repack Major
@callstack/repack-dev-server Major
@callstack/repack-init Major
@callstack/repack-plugin-nativewind Major
@callstack/repack-plugin-expo-modules Major

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@vercel

vercel Bot commented Aug 22, 2025 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Preview Comments Updated (UTC)
repack-website Ready Ready Preview Comment Aug 22, 2025 10:50am

@jbroma
jbroma merged commit 58a459f into main Aug 22, 2025
5 checks passed
@jbroma
jbroma deleted the feat/allow-passing-options-to-reanimated branch August 22, 2025 10:50

This branch was successfully deployed

1 active deployment
Preview — 5a0638ce Deployed Aug 22, 2025 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant